Skip to content

Update kubernetes monorepo to v0.37.1 - #46

Merged
ske-prow[bot] merged 1 commit into
mainfrom
renovate/kubernetes-monorepo
Sep 28, 2026
Merged

ske-prow[bot] merged 1 commit into
mainfrom
renovate/kubernetes-monorepo

Conversation

@ske-renovate-operator

@ske-renovate-operator ske-renovate-operator Bot commented Aug 29, 2026 •

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Change Age Confidence
k8s.io/api v0.37.0 → v0.37.1 age confidence
k8s.io/apimachinery v0.37.0 → v0.37.1 age confidence
k8s.io/client-go v0.37.0 → v0.37.1 age confidence

Release Notes

kubernetes/api (k8s.io/api)

v0.37.1

Compare Source

kubernetes/apimachinery (k8s.io/apimachinery)

v0.37.1

Compare Source

kubernetes/client-go (k8s.io/client-go)

v0.37.1

Compare Source


Configuration

📅 Schedule: (UTC)

  • Branch creation
    • At any time (no schedule defined)
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Enabled.

♻ Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about these updates again.


  • If you want to rebase/retry this PR, check this box

This PR has been generated by Mend Renovate CLI.

@ske-renovate-operator ske-renovate-operator Bot added the kind/upgrade Upgrading components or dependencies (e.g., by renovate) label Aug 29, 2026
@ske-renovate-operator
ske-renovate-operator Bot requested a review from a team August 29, 2026 20:40
@ske-prow

ske-prow Bot commented Aug 29, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign crigertg for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@ske-prow ske-prow Bot added the size/L Denotes a PR that changes 100-499 lines, ignoring generated files. label Aug 29, 2026
@ske-renovate-operator
ske-renovate-operator Bot force-pushed the renovate/kubernetes-monorepo branch from 41134e1 to ec0140e Compare August 31, 2026 00:44
@ske-renovate-operator ske-renovate-operator Bot changed the title Update kubernetes monorepo to v0.37.0 Update module k8s.io/apimachinery to v0.37.0 Sep 3, 2026
@ske-renovate-operator ske-renovate-operator Bot changed the title Update module k8s.io/apimachinery to v0.37.0 Update kubernetes monorepo to v0.37.0 Sep 3, 2026
@ske-renovate-operator
ske-renovate-operator Bot force-pushed the renovate/kubernetes-monorepo branch 2 times, most recently from daafa4d to eae47fd Compare September 8, 2026 16:37
@ske-renovate-operator
ske-renovate-operator Bot force-pushed the renovate/kubernetes-monorepo branch 2 times, most recently from 2954ce2 to d3b042a Compare September 27, 2026 00:40
@ske-renovate-operator ske-renovate-operator Bot changed the title Update kubernetes monorepo to v0.37.0 Update kubernetes monorepo to v0.37.1 Sep 27, 2026
@ske-renovate-operator
ske-renovate-operator Bot force-pushed the renovate/kubernetes-monorepo branch from d3b042a to c232ff4 Compare September 28, 2026 06:57
@ske-prow ske-prow Bot added size/S Denotes a PR that changes 10-29 lines, ignoring generated files. and removed size/L Denotes a PR that changes 100-499 lines, ignoring generated files. labels Sep 28, 2026
@ske-renovate-operator
ske-renovate-operator Bot force-pushed the renovate/kubernetes-monorepo branch from c232ff4 to b1337f7 Compare September 28, 2026 07:48
@ske-renovate-operator

Copy link
Copy Markdown
Contributor Author

/label skip-review

@ske-prow ske-prow Bot added the skip-review Indicates a PR is trusted, used by tide for auto-merging PRs. label Sep 28, 2026
@ske-prow
ske-prow Bot merged commit 6dc8ed1 into main Sep 28, 2026
4 checks passed
@ske-prow
ske-prow Bot deleted the renovate/kubernetes-monorepo branch September 28, 2026 07:51
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

kind/upgrade Upgrading components or dependencies (e.g., by renovate) size/S Denotes a PR that changes 10-29 lines, ignoring generated files. skip-review Indicates a PR is trusted, used by tide for auto-merging PRs.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants